Skip to content

Remove deprecated markdown-remark APIs - #9182

Merged
bluwy merged 1 commit into
nextfrom
remove-deprecate-markdown-stuff
Nov 27, 2023
Merged

bluwy merged 1 commit into
nextfrom
remove-deprecate-markdown-stuff

Conversation

@bluwy

@bluwy bluwy commented Nov 23, 2023

Copy link
Copy Markdown
Member

Changes

These APIs are not used by us anymore. They're only kept so we can remove them in the next markdown-remark major, like now in the next branch

Testing

Existing tests should work.

Docs

n/a. internal refactor. Added a changeset though so we can track this change in the future.

@changeset-bot

changeset-bot Bot commented Nov 23, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 44423d9

The changes in this PR will be included in the next version bump.

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions github-actions Bot added feat: markdown Related to Markdown (scope) pr: docs A PR that includes documentation for review labels Nov 23, 2023
@bluwy
bluwy merged commit c795364 into next Nov 27, 2023
@bluwy
bluwy deleted the remove-deprecate-markdown-stuff branch November 27, 2023 16:32
@astrobot-houston astrobot-houston mentioned this pull request Dec 5, 2023
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

feat: markdown Related to Markdown (scope) pr: docs A PR that includes documentation for review

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants