Skip to content

Allow AlpineJS syntax extensions in Markdown - #3554

Merged
natemoo-re merged 2 commits into
withastro:mainfrom
hippotastic:fix/allow-alpinejs-syntax-in-md
Jun 8, 2022
Merged

natemoo-re merged 2 commits into
withastro:mainfrom
hippotastic:fix/allow-alpinejs-syntax-in-md

Conversation

@hippotastic

Copy link
Copy Markdown
Contributor

Changes

Testing

  • Added new test cases to cover all relevant cases of the AlpineJS syntax based on their docs.
  • Successfully ran all tests locally.
  • Built Astro Docs with the new packages.
  • Built a sample Astro project that contains the AlpineJS example code from their homepage in a Markdown file, and the resulting page worked.

Docs

  • Not a visible change.

@changeset-bot

changeset-bot Bot commented Jun 8, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 85b059d

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@astrojs/markdown-remark Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions github-actions Bot added the feat: markdown Related to Markdown (scope) label Jun 8, 2022
@hippotastic

Copy link
Copy Markdown
Contributor Author

@natemoo-re PTAL :)

@natemoo-re natemoo-re left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks great! Just had a few questions.

Comment thread packages/markdown/remark/src/mdxjs.ts Outdated
Comment thread packages/markdown/remark/src/mdxjs.ts

@natemoo-re natemoo-re left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM!

@natemoo-re
natemoo-re merged commit c549f16 into withastro:main Jun 8, 2022
@github-actions github-actions Bot mentioned this pull request Jun 8, 2022
@hippotastic
hippotastic deleted the fix/allow-alpinejs-syntax-in-md branch June 8, 2022 22:52
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

feat: markdown Related to Markdown (scope)

Projects

None yet

Development

Successfully merging this pull request may close these issues.

🐛 BUG: Alpinejs bind shorthands (ex. :width="<expression>") in markdown error dev and build

2 participants