Skip to content

chore: changeset for failed vscode publish - #15825

Merged
Princesseuh merged 2 commits into
mainfrom
fix/vscode-changeset
Mar 10, 2026
Merged

chore: changeset for failed vscode publish#15825
Princesseuh merged 2 commits into
mainfrom
fix/vscode-changeset

Conversation

@Princesseuh

@Princesseuh Princesseuh commented Mar 10, 2026

Copy link
Copy Markdown
Member

Changes

  • Fixes failed publish

Testing

Docs

@changeset-bot

changeset-bot Bot commented Mar 10, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 687f499

The changes in this PR will be included in the next version bump.

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@Princesseuh
Princesseuh marked this pull request as ready for review March 10, 2026 12:38
@Princesseuh Princesseuh changed the title chore: changeset chore: changeset for failed vscode publish Mar 10, 2026
Comment thread .changeset/cuddly-trains-pay.md Outdated
Co-authored-by:  Matthew Phillips <matthewphillips@cloudflare.com>
@Princesseuh
Princesseuh merged commit 2cfe935 into main Mar 10, 2026
2 checks passed
@Princesseuh
Princesseuh deleted the fix/vscode-changeset branch March 10, 2026 12:42
@astrobot-houston astrobot-houston mentioned this pull request Mar 10, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants