Skip to content

chore: Try publishing VS Code another time - #14988

Merged
Princesseuh merged 1 commit into
mainfrom
fix/vscode-release-last-final-final-3
Dec 10, 2025
Merged

chore: Try publishing VS Code another time#14988
Princesseuh merged 1 commit into
mainfrom
fix/vscode-release-last-final-final-3

Conversation

@Princesseuh

Copy link
Copy Markdown
Member

Changes

We updated the token and it should work now

Testing

N/A

Docs

N/A

@changeset-bot

changeset-bot Bot commented Dec 9, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 7bef210

The changes in this PR will be included in the next version bump.

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@delucis delucis left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Good luck!

@Princesseuh
Princesseuh merged commit a3a20d8 into main Dec 10, 2025
2 checks passed
@Princesseuh
Princesseuh deleted the fix/vscode-release-last-final-final-3 branch December 10, 2025 13:34
@astrobot-houston astrobot-houston mentioned this pull request Dec 10, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants