Skip to content

fix(ci): Try building package before deleting dependencies for vscode - #14916

Merged
Princesseuh merged 2 commits into
mainfrom
fix/vscode-release-last-final-please
Nov 27, 2025
Merged

fix(ci): Try building package before deleting dependencies for vscode#14916
Princesseuh merged 2 commits into
mainfrom
fix/vscode-release-last-final-please

Conversation

@Princesseuh

Copy link
Copy Markdown
Member

Changes

Building the ts-plugin after deleting node_modules obviously doesn't work, but it doesn't matter because it's already been built!

Testing

We hope!

Docs

N/A

@changeset-bot

changeset-bot Bot commented Nov 27, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 5873a24

The changes in this PR will be included in the next version bump.

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions github-actions Bot added the 🚨 action Modifies GitHub Actions label Nov 27, 2025
@Princesseuh Princesseuh changed the title fix(ci): Do not prebuild ts-plugin after deleting node_modules in vscode publishing fix(ci): Try building package before deleting dependencies for vscode Nov 27, 2025
@Princesseuh
Princesseuh marked this pull request as ready for review November 27, 2025 04:17
@Princesseuh
Princesseuh merged commit f00aa2b into main Nov 27, 2025
24 checks passed
@Princesseuh
Princesseuh deleted the fix/vscode-release-last-final-please branch November 27, 2025 16:22
@astrobot-houston astrobot-houston mentioned this pull request Nov 27, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

🚨 action Modifies GitHub Actions

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants