Skip to content

chore: changeset - #14909

Merged
Princesseuh merged 1 commit into
mainfrom
fix/vscode-release2
Nov 26, 2025
Merged

chore: changeset#14909
Princesseuh merged 1 commit into
mainfrom
fix/vscode-release2

Conversation

@Princesseuh

@Princesseuh Princesseuh commented Nov 26, 2025

Copy link
Copy Markdown
Member

Changes

Just an empty changeset to do a VS Code release after the CI fix

Testing

N/A

Docs

N/A

@changeset-bot

changeset-bot Bot commented Nov 26, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 47e0a6e

The changes in this PR will be included in the next version bump.

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@Princesseuh
Princesseuh merged commit 629d86b into main Nov 26, 2025
2 checks passed
@Princesseuh
Princesseuh deleted the fix/vscode-release2 branch November 26, 2025 15:11
@astrobot-houston astrobot-houston mentioned this pull request Nov 26, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants