Skip to content

fix: preserve HAST properties in image processing - #14902

Merged
florian-lefebvre merged 3 commits into
withastro:mainfrom
tuyuritio:fix/hast-preserved-props
Dec 4, 2025
Merged

florian-lefebvre merged 3 commits into
withastro:mainfrom
tuyuritio:fix/hast-preserved-props

Conversation

@tuyuritio

@tuyuritio tuyuritio commented Nov 25, 2025

Copy link
Copy Markdown
Contributor

Changes

Testing

  • Add a new test case in remark-collect-images.test.js
  • markdown/remark tests passed

Docs

N/A

@changeset-bot

changeset-bot Bot commented Nov 25, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: eb29ab2

The changes in this PR will be included in the next version bump.

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions github-actions Bot added the feat: markdown Related to Markdown (scope) label Nov 25, 2025
@florian-lefebvre

Copy link
Copy Markdown
Member

Looks reasonable to me, @delucis wdyt?

@florian-lefebvre florian-lefebvre left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM, thank you!

Comment thread .changeset/pink-experts-feel.md Outdated
Co-authored-by: Florian Lefebvre <contact@florian-lefebvre.dev>
@florian-lefebvre
florian-lefebvre merged commit d8305f8 into withastro:main Dec 4, 2025
22 of 24 checks passed
@astrobot-houston astrobot-houston mentioned this pull request Dec 4, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

feat: markdown Related to Markdown (scope)

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Unable to add class to markdown images via plugin

2 participants