Skip to content

Fix description for "Content-intellisense" VSC extension setting - #14710

Merged
Princesseuh merged 3 commits into
withastro:mainfrom
RomanHauksson:main
Nov 26, 2025
Merged

Fix description for "Content-intellisense" VSC extension setting#14710
Princesseuh merged 3 commits into
withastro:mainfrom
RomanHauksson:main

Conversation

@RomanHauksson

Copy link
Copy Markdown
Contributor

Changes

  • The description of the "Content-intellisense" setting in the VS Code extension refers to the relevant configuration file option as "experimental.contentCollectionIntellisense". This is incorrect; I changed it to say "experimental.contentIntellisense".
  • Grammar fix ("require" → "requires").

(This is my first pull request to Astro, apologies in advanced if I missed anything!)

Testing

No tests were added, this is just a documentation fix.

Docs

No docs were added, this is just a documentation fix.

@changeset-bot

changeset-bot Bot commented Nov 3, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: c0ee9c2

The changes in this PR will be included in the next version bump.

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@Princesseuh Princesseuh left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks great! Thank you

@Princesseuh

Copy link
Copy Markdown
Member

Just a heads up that we're currently trying to figure out the release process for the VS Code extension from this repo, so this unfortunately won't be able to be merged and released just now! Thank you again for contributing!

@RomanHauksson

Copy link
Copy Markdown
Contributor Author

Yep np! Thanks for reviewing.

@Princesseuh
Princesseuh merged commit c6903cd into withastro:main Nov 26, 2025
24 checks passed
@astrobot-houston astrobot-houston mentioned this pull request Nov 26, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants