Skip to content

Add QTT as adopter - #536

Merged
volcano-sh-bot merged 1 commit into
masterfrom
add_qtt
Nov 19, 2019
Merged

Add QTT as adopter#536
volcano-sh-bot merged 1 commit into
masterfrom
add_qtt

Conversation

@k82cn

@k82cn k82cn commented Nov 19, 2019

Copy link
Copy Markdown
Member

No description provided.

@volcano-sh-bot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: k82cn

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k82cn

k82cn commented Nov 19, 2019

Copy link
Copy Markdown
Member Author

/cc @yuzhaojing

@volcano-sh-bot volcano-sh-bot added approved Indicates a PR has been approved by an approver from all required OWNERS files. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Nov 19, 2019
@volcano-sh-bot

Copy link
Copy Markdown
Contributor

@k82cn: GitHub didn't allow me to request PR reviews from the following users: yuzhaojing.

Note that only volcano-sh members and repo collaborators can review this PR, and authors cannot review their own PRs.

Details

In response to this:

/cc @yuzhaojing

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@TravisBuddy

Copy link
Copy Markdown

Hey @k82cn,
Your changes look good to me!

View build log

TravisBuddy Request Identifier: 6fa10bf0-0a9d-11ea-adc5-254fad528e65

@k82cn k82cn added the lgtm Indicates that a PR is ready to be merged. label Nov 19, 2019
@volcano-sh-bot
volcano-sh-bot merged commit f453737 into master Nov 19, 2019
@volcano-sh-bot
volcano-sh-bot deleted the add_qtt branch November 19, 2019 13:32
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. retest-not-required-docs-only size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants