Skip to content

chore: Updated CI to skip expensive actions when not needed - #2184

Merged
jsumners-nr merged 1 commit into
newrelic:mainfrom
jsumners-nr:faster-ci
May 7, 2024
Merged

chore: Updated CI to skip expensive actions when not needed#2184
jsumners-nr merged 1 commit into
newrelic:mainfrom
jsumners-nr:faster-ci

Conversation

@jsumners-nr

Copy link
Copy Markdown
Contributor

This PR changes ci.yml to only run expensive linting and testing jobs if any JavaScript files have changed. In other words, docs changes or other repo maintenance will not trigger the long running CI jobs.

Note: we need to update the main branch protection rule to require the all-clear job to pass and remove the direct requirements on specific linting/testing jobs.

@jsumners-nr
jsumners-nr marked this pull request as ready for review May 7, 2024 13:58
@jsumners-nr
jsumners-nr merged commit 8639fb9 into newrelic:main May 7, 2024
@jsumners-nr
jsumners-nr deleted the faster-ci branch May 7, 2024 15:55
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

Archived in project

Development

Successfully merging this pull request may close these issues.

3 participants